Skip to content

Added google_artifact_registry_repository_iam_member list resource - #18931

Draft
malhotrasagar2212 wants to merge 1 commit into
GoogleCloudPlatform:mainfrom
malhotrasagar2212:google_artifact_registry_repository_iam_member_list
Draft

Added google_artifact_registry_repository_iam_member list resource#18931
malhotrasagar2212 wants to merge 1 commit into
GoogleCloudPlatform:mainfrom
malhotrasagar2212:google_artifact_registry_repository_iam_member_list

Conversation

@malhotrasagar2212

Copy link
Copy Markdown
Collaborator

Release Note Template for Downstream PRs (will be copied)
Added google_artifact_registry_repository_iam_member list resource
See Write release notes for guidance.


@modular-magician

modular-magician commented Sep 9, 2026

Copy link
Copy Markdown
Collaborator

Hi there, I'm the Modular magician. I've detected the following information about your changes for commit 08d1f12:

Diff report

Your PR generated the following diffs in downstream repositories:

Repository Diff Link Changes
google provider View Diff 4 files changed, 229 insertions(+)
google-beta provider View Diff 4 files changed, 229 insertions(+)

Test report

Analytics

Total Tests Passed Skipped Affected
58 47 10 1
Affected Service Packages
  • artifactregistry

Learn how VCR tests work


Step 1: Replaying Mode

Action taken

Found 1 affected test(s) by replaying old test recordings. Starting RECORDING based on the most recent commit.

Click here to see the affected tests
  • TestAccArtifactRegistryRepositoryIamMemberListQuery_generated

View the replaying VCR build log


Step 2: Recording Mode

Recording Mode Replaying Rerun Test Name
✅ Log TestAccArtifactRegistryRepositoryIamMemberListQuery_generated

🟢 All tests passed!

View the recording VCR build log or the debug logs folder for detailed results.

@malhotrasagar2212, @bobcatfish VCR tests complete for 08d1f12!

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants