Skip to content

Bump gopkg.in/ini.v1 from 1.67.0 to 1.67.3 - #1026

Open
dependabot[bot] wants to merge 1 commit into
masterfrom
dependabot/go_modules/gopkg.in/ini.v1-1.67.3
Open

Bump gopkg.in/ini.v1 from 1.67.0 to 1.67.3#1026
dependabot[bot] wants to merge 1 commit into
masterfrom
dependabot/go_modules/gopkg.in/ini.v1-1.67.3

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Jul 28, 2026

Copy link
Copy Markdown
Contributor

Bumps gopkg.in/ini.v1 from 1.67.0 to 1.67.3.

@dependabot dependabot Bot added the dependencies Pull requests that update a dependency file label Jul 28, 2026
@google-oss-prow

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: dependabot[bot]
Once this PR has been reviewed and has the lgtm label, please assign burov for approval. For more information see the Kubernetes Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@google-oss-prow

Copy link
Copy Markdown

Hi @dependabot[bot]. Thanks for your PR.

I'm waiting for a GoogleCloudPlatform member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work. Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@burov

burov commented Jul 28, 2026

Copy link
Copy Markdown
Member

/gcbrun

@dependabot
dependabot Bot force-pushed the dependabot/go_modules/gopkg.in/ini.v1-1.67.3 branch 2 times, most recently from 65c45cc to faa47c7 Compare July 31, 2026 12:54
Bumps gopkg.in/ini.v1 from 1.67.0 to 1.67.3.

---
updated-dependencies:
- dependency-name: gopkg.in/ini.v1
  dependency-version: 1.67.3
  dependency-type: indirect
  update-type: version-update:semver-patch
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot
dependabot Bot force-pushed the dependabot/go_modules/gopkg.in/ini.v1-1.67.3 branch from faa47c7 to f80a21b Compare August 11, 2026 09:11
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

dependencies Pull requests that update a dependency file needs-ok-to-test size/XS

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant