Bump github.com/GoogleCloudPlatform/opentelemetry-operations-go/exporter/metric from 0.54.0 to 0.59.0 - #1027
Conversation
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: dependabot[bot] The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
Hi @dependabot[bot]. Thanks for your PR. I'm waiting for a GoogleCloudPlatform member to verify that this patch is reasonable to test. If it is, they should reply with Once the patch is verified, the new status will be reflected by the I understand the commands that are listed here. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
ec64198 to
f28e506
Compare
…ter/metric Bumps [github.com/GoogleCloudPlatform/opentelemetry-operations-go/exporter/metric](https://github.com/GoogleCloudPlatform/opentelemetry-operations-go) from 0.54.0 to 0.59.0. - [Release notes](https://github.com/GoogleCloudPlatform/opentelemetry-operations-go/releases) - [Commits](GoogleCloudPlatform/opentelemetry-operations-go@v0.54.0...v0.59.0) --- updated-dependencies: - dependency-name: github.com/GoogleCloudPlatform/opentelemetry-operations-go/exporter/metric dependency-version: 0.59.0 dependency-type: indirect update-type: version-update:semver-minor ... Signed-off-by: dependabot[bot] <support@github.com>
f28e506 to
09a2e84
Compare
Bumps github.com/GoogleCloudPlatform/opentelemetry-operations-go/exporter/metric from 0.54.0 to 0.59.0.
Release notes
Sourced from github.com/GoogleCloudPlatform/opentelemetry-operations-go/exporter/metric's releases.
... (truncated)
Commits
d84f5e9Prepare release 0.59.0/1.35.0 (#1183)c0846cfUpdate module google.golang.org/grpc to v1.82.1 [SECURITY] (#1184)6dd64feupdate golang.org/x/crypto (#1182)a011b90Adds support for resolving OpenTelemetry authentication extensions (#1181)53b33bePrepare release v1.34.0/v0.58.0 (#1178)411c171googlemanagedprometheus: Adddestination_project_quota(#1175)059f723Prepare for 1.33.0/v0.57.0 release (#1173)12f7c67Add support for cloud run worker pools (#1112)631aa83chore: update E2E tests to use centralized async cleanup v0.22.0 (#1171)6c5e750exporter/trace: sanitize invalid UTF-8 in span fields (#1141)