Skip to content

Remove @anup-kodlekere from CODEOWNERS - #54

Merged
janani66 merged 1 commit into
mainfrom
offboard-anup
Sep 21, 2026
Merged

janani66 merged 1 commit into
mainfrom
offboard-anup

Conversation

@anup-kodlekere

Copy link
Copy Markdown
Contributor

Offboard anup-kodlekere. LWD with the team.

Offboard anup-kodlekere. LWD with the team. 

Signed-off-by: anup-kodlekere <anup.kodlekere@gmail.com>

@janani66 janani66 left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM

@janani66
janani66 merged commit 4952fe5 into main Sep 21, 2026
7 checks passed
@janani66
janani66 deleted the offboard-anup branch September 21, 2026 12:07
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants