Skip to content

chore(deps): update dawidd6/action-send-mail action to v20 - autoclosed - #1078

Closed
renovate[bot] wants to merge 1 commit into
mainfrom
renovate/dawidd6-action-send-mail-20.x
Closed

renovate[bot] wants to merge 1 commit into
mainfrom
renovate/dawidd6-action-send-mail-20.x

Conversation

@renovate

@renovate renovate Bot commented Sep 8, 2026

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Type Update Change
dawidd6/action-send-mail action major v18v20

Release Notes

dawidd6/action-send-mail (dawidd6/action-send-mail)

v20

Compare Source

BREAKING CHANGE

HTTP_PROXY is no longer respected, use SMTP_PROXY environment variable.

What's Changed

New Contributors

Full Changelog: dawidd6/action-send-mail@v19...v20

v19

Compare Source

What's Changed

Full Changelog: dawidd6/action-send-mail@v18...v19


Configuration

📅 Schedule: (UTC)

  • Branch creation
    • At any time (no schedule defined)
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, check this box

This PR was generated by Mend Renovate. View the repository job log.

@renovate renovate Bot added the help wanted Extra attention is needed label Sep 8, 2026
@ti-chi-bot

ti-chi-bot Bot commented Sep 8, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign dillon-zheng for approval. For more information see the Code Review Process.
Please ensure that each of them provides their approval before proceeding.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@ti-chi-bot ti-chi-bot Bot added the size/XS label Sep 8, 2026
@renovate renovate Bot changed the title chore(deps): update dawidd6/action-send-mail action to v20 chore(deps): update dawidd6/action-send-mail action to v20 - autoclosed Sep 10, 2026
@renovate renovate Bot closed this Sep 10, 2026
@renovate
renovate Bot deleted the renovate/dawidd6-action-send-mail-20.x branch September 10, 2026 23:12
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

help wanted Extra attention is needed size/XS

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants