Skip to content

chore(deps): update dependency just to v1.58.0 - #1081

Open
renovate[bot] wants to merge 1 commit into
mainfrom
renovate/just-1.x
Open

renovate[bot] wants to merge 1 commit into
mainfrom
renovate/just-1.x

Conversation

@renovate

@renovate renovate Bot commented Sep 16, 2026

Copy link
Copy Markdown
Contributor

ℹ️ Note

This PR body was truncated due to platform limits.

This PR contains the following updates:

Package Type Update Change
just uses-with minor 1.34.01.58.0

Release Notes

casey/just (just)

v1.58.0

Compare Source

Changed
  • Surround interoplation expressions with spaces when formatting (#​3685 by casey)
  • Accept multiple ,-separated values in --justfile-name (#​3676 by casey)
Added
Misc

v1.57.0

Compare Source

Changed
Fixed
  • Don't reevaluate lazy assignments in user-defined functions (#​3654 by casey)
  • Don't print parenthesis around dependencies without arguments (#​3653 by casey)
  • Accept override paths containing module aliases (#​3652 by casey)
  • Fix line number in error message after multi-line interpolation (#​3646 by casey)
  • Don't execute dotenv commands during dry run (#​3644 by casey)
  • Only run dotenv-command in root module (#​3643 by casey)
  • Detect module cycle when path uses .. (#​3641 by casey)
  • Recipes with [arg(min)] greater than zero cannot be default (#​3639 by casey)
  • Avoid panic in ambiguous module file error (#​3638 by casey)
  • Consider value when checking for setting conflicts (#​3627 by casey)
  • Ignore disabled indentation settings (#​3618 by casey)
  • Reject duplicate function parameters (#​3611 by casey)
  • Allow defaulted option before required positional parameter (#​3609 by casey)
  • Truncate cache entry before writing (#​3603 by casey)
  • Ignore absent optional modules under --allow-missing (#​3601 by casey)
Added
Misc

v1.56.0

Compare Source

Fixed
Changed
Added
Misc

v1.55.1

Compare Source

Fixed
  • Fix panic when empty path is passed to --show and --usage (#​3511 by casey)
Misc

v1.55.0

Compare Source

Fixed
Changed
Added
Misc

v1.54.0

Compare Source

Fixed
  • Forbid duplicate attribute keys (#​3454 by casey)
  • Clear signal handler when infallible line catches signal (#​3444 by casey)
Changed
Added
Cached Recipes
Misc

v1.53.0

Compare Source

Added
Changed
Lists
Misc

v1.52.0

Compare Source

Added
Changed
  • Disable aliases that depend on absent modules (#​3358 by casey)
  • Print formatted justfile to stdout when source is stdin (#​3356 by casey)
  • Allow search directory with more subcommands (#​3353 by casey)
  • Disable recipes that depend on absent modules (#​3352 by casey)
Misc

v1.51.0

Compare Source

Added
Changed
  • Allow [env] attribute to take expressions (#​3329 by casey)
  • Allow using expressions with [working-directory] (#​3326 by casey)
  • Read justfile from standard input with --justfile - (#​3325 by casey)
  • Make parent_directory() of bare filename return . (#​3313 by casey)
  • Make [env] override module-level exports (#​3312 by casey)
Fixed
Misc

v1.50.0

Compare Source

Added
Changed
  • Allow [env] attribute to take expressions (#​3329 by casey)
  • Allow using expressions with [working-directory] (#​3326 by casey)
  • Read justfile from standard input with --justfile - (#​3325 by casey)
  • Make parent_directory() of bare filename return . (#​3313 by casey)
  • Make [env] override module-level exports (#​3312 by casey)
Fixed
Misc

v1.49.0

Compare Source

Added
Fixed
Misc

v1.48.1

Compare Source

  • Fix bash completion of command lines containing : (#​3205 by casey)
  • Call zsh completion function when first auto-loaded (#​3199 by casey)
  • Fix --set missing value error message interpolation (#​3198 by cuiweixie)
  • Add #compdef to zsh completion script (#​3197 by casey)
  • Add instructions for alias completions in bash ([#&#8203

Important

✂ PR body was truncated to here.


Configuration

📅 Schedule: (UTC)

  • Branch creation
    • At any time (no schedule defined)
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, check this box

This PR was generated by Mend Renovate. View the repository job log.

@renovate renovate Bot added the help wanted Extra attention is needed label Sep 16, 2026
@ti-chi-bot

ti-chi-bot Bot commented Sep 16, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign dillon-zheng for approval. For more information see the Code Review Process.
Please ensure that each of them provides their approval before proceeding.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@ti-chi-bot ti-chi-bot Bot added the size/XS label Sep 16, 2026
@renovate
renovate Bot force-pushed the renovate/just-1.x branch from aeef71a to eb6a7ff Compare September 16, 2026 20:02
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

help wanted Extra attention is needed size/XS

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants