Skip to content

chore(deps): update ghcr.io/pingcap-qe/bases/pingcap-base docker tag to v1.12.3 - #4923

Merged
ti-chi-bot[bot] merged 1 commit into
mainfrom
renovate/ghcr.io-pingcap-qe-bases-pingcap-base-1.12.x
Aug 19, 2026
Merged

ti-chi-bot[bot] merged 1 commit into
mainfrom
renovate/ghcr.io-pingcap-qe-bases-pingcap-base-1.12.x

Conversation

@renovate

@renovate renovate Bot commented Aug 17, 2026

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Type Update Change
ghcr.io/pingcap-qe/bases/pingcap-base final patch v1.12.2v1.12.3

Warning

Some dependencies could not be looked up. Check the Dependency Dashboard for more information.


Configuration

📅 Schedule: (UTC)

  • Branch creation
    • At any time (no schedule defined)
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, check this box

This PR was generated by Mend Renovate. View the repository job log.

@renovate
renovate Bot force-pushed the renovate/ghcr.io-pingcap-qe-bases-pingcap-base-1.12.x branch from 8543f27 to 1f655ec Compare August 18, 2026 20:04
@ti-chi-bot ti-chi-bot Bot added the lgtm label Aug 19, 2026
@ti-chi-bot

ti-chi-bot Bot commented Aug 19, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: wuhuizuo

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@ti-chi-bot

ti-chi-bot Bot commented Aug 19, 2026

Copy link
Copy Markdown

[LGTM Timeline notifier]

Timeline:

  • 2026-08-19 09:46:07.592952664 +0000 UTC m=+63002.764046777: ☑️ agreed by wuhuizuo.

@ti-chi-bot ti-chi-bot Bot added the approved label Aug 19, 2026
@ti-chi-bot
ti-chi-bot Bot merged commit 50dc203 into main Aug 19, 2026
11 checks passed
@ti-chi-bot
ti-chi-bot Bot deleted the renovate/ghcr.io-pingcap-qe-bases-pingcap-base-1.12.x branch August 19, 2026 09:48
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

Status: ✅ Done

Development

Successfully merging this pull request may close these issues.

1 participant