chore(deps): update dependency nodemailer to v10 - #5209
renovate[bot] wants to merge 1 commit into
Conversation
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
Hi @renovate[bot]. Thanks for your PR. I'm waiting for a PingCAP-QE member to verify that this patch is reasonable to test. If it is, they should reply with Once the patch is verified, the new status will be reflected by the I understand the commands that are listed here. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
169922c to
2c62154
Compare
2c62154 to
521fc17
Compare
This PR contains the following updates:
^6.9.16->^10.0.0Warning
Some dependencies could not be looked up. Check the Dependency Dashboard for more information.
Release Notes
nodemailer/nodemailer (nodemailer)
v10.0.10Compare Source
Bug Fixes
v10.0.9Compare Source
Bug Fixes
v10.0.8Compare Source
Bug Fixes
v10.0.7Compare Source
Bug Fixes
v10.0.6Compare Source
Bug Fixes
v10.0.5Compare Source
Bug Fixes
v10.0.4Compare Source
Bug Fixes
v10.0.3Compare Source
Bug Fixes
v10.0.2Compare Source
Bug Fixes
v10.0.1Compare Source
Bug Fixes
v10.0.0Compare Source
⚠ BREAKING CHANGES
Features
Bug Fixes
v9.1.1Compare Source
Bug Fixes
v9.1.0Compare Source
Features
Bug Fixes
v9.0.6Compare Source
Bug Fixes
v9.0.5Compare Source
Bug Fixes
v9.0.4Compare Source
Bug Fixes
v9.0.3Compare Source
Bug Fixes
v9.0.2Compare Source
Bug Fixes
v9.0.1Compare Source
Bug Fixes
v9.0.0Compare Source
⚠ BREAKING CHANGES
tlsoption).Bug Fixes
v8.0.11Compare Source
Bug Fixes
v8.0.10Compare Source
Bug Fixes
v8.0.9Compare Source
Bug Fixes
v8.0.8Compare Source
Bug Fixes
v8.0.7Compare Source
Bug Fixes
v8.0.6Compare Source
Bug Fixes
v8.0.5Compare Source
Bug Fixes
v8.0.4Compare Source
Bug Fixes
v8.0.3Compare Source
Bug Fixes
v8.0.2Compare Source
Bug Fixes
v8.0.1Compare Source
Bug Fixes
v8.0.0Compare Source
⚠ BREAKING CHANGES
Bug Fixes
v7.0.13Compare Source
Bug Fixes
v7.0.12Compare Source
Bug Fixes
v7.0.11Compare Source
Bug Fixes
v7.0.10Compare Source
Bug Fixes
v7.0.9Compare Source
Bug Fixes
v7.0.7Compare Source
Bug Fixes
v7.0.6Compare Source
Bug Fixes
v7.0.5Compare Source
Bug Fixes
v7.0.4Compare Source
Bug Fixes
v7.0.3Compare Source
Bug Fixes
v7.0.2Compare Source
Bug Fixes
v7.0.1Compare Source
Bug Fixes
v7.0.0Compare Source
⚠ BREAKING CHANGES
Features
Configuration
📅 Schedule: (UTC)
🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.
🔕 Ignore: Close this PR and you won't be reminded about this update again.
This PR was generated by Mend Renovate. View the repository job log.