Skip to content

Bump k8s.io/apimachinery from v0.29.0 to v0.36.2 - #49

Open
tmcgahern wants to merge 1 commit into
aws-controllers-k8s:mainfrom
tmcgahern:update-apimachinery-v0.36.2
Open

Bump k8s.io/apimachinery from v0.29.0 to v0.36.2#49
tmcgahern wants to merge 1 commit into
aws-controllers-k8s:mainfrom
tmcgahern:update-apimachinery-v0.36.2

Conversation

@tmcgahern

Copy link
Copy Markdown

Description of changes:

Update k8s.io/apimachinery version from v0.29.0 to latest version v0.36.2

Commands executed to update

go get -u k8s.io/apimachinery
go mod tidy

By submitting this pull request, I confirm that my contribution is made under the terms of the Apache 2.0 license.

@ack-prow
ack-prow Bot requested a review from knottnt July 17, 2026 11:44
@ack-prow

ack-prow Bot commented Jul 17, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: tmcgahern
Once this PR has been reviewed and has the lgtm label, please assign michaelhtm for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@ack-prow
ack-prow Bot requested a review from michaelhtm July 17, 2026 11:44
@ack-prow ack-prow Bot added the needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test. label Jul 17, 2026
@ack-prow

ack-prow Bot commented Jul 17, 2026

Copy link
Copy Markdown

Hi @tmcgahern. Thanks for your PR.

I'm waiting for a aws-controllers-k8s member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work.

Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@knottnt knottnt self-assigned this Jul 23, 2026
@knottnt

knottnt commented Jul 23, 2026

Copy link
Copy Markdown
Contributor

/test all

@ack-prow

ack-prow Bot commented Jul 23, 2026

Copy link
Copy Markdown

@tmcgahern: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
verify-attribution 4b8d2b8 link false /test verify-attribution

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants