Skip to content

feat: make ARN unique identifier for AWS Resource Shares - #58

Open
ramasai1 wants to merge 1 commit into
aws-controllers-k8s:mainfrom
ramasai1:make-arn-unique-identifier
Open

feat: make ARN unique identifier for AWS Resource Shares#58
ramasai1 wants to merge 1 commit into
aws-controllers-k8s:mainfrom
ramasai1:make-arn-unique-identifier

Conversation

@ramasai1

Copy link
Copy Markdown

We're using ACK team based CARM, where a single namespace is mapped to a team. Each team is creating a ResourceShare CR with the same name in their VPC, but the RAM controller is querying for ResourceShares via the .spec.Name key. This makes it such that all of the ResourceShare objects in all namespaces end up getting the same ARN, and the resources shared get overwritten.

We should be matching on ARN and not on the name, since thats a unique identifier for the share. If the ID of the share was available as a standalone field (similar to VPC ID) then things would have been easier.

@ack-prow
ack-prow Bot requested review from a-hilaly and jlbutler July 20, 2026 18:02
@ack-prow

ack-prow Bot commented Jul 20, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: ramasai1
Once this PR has been reviewed and has the lgtm label, please assign sapphirew for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@ack-prow ack-prow Bot added the needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test. label Jul 20, 2026
@ack-prow

ack-prow Bot commented Jul 20, 2026

Copy link
Copy Markdown

Hi @ramasai1. Thanks for your PR.

I'm waiting for a aws-controllers-k8s member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work.

Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@ramasai1

Copy link
Copy Markdown
Author

The existing tests should pass - I am expecting this to be a transparent change for both resource shares that exist and new resource shares being created.

@ramasai1

Copy link
Copy Markdown
Author

Hello @a-hilaly - looking to get some early thoughts from you about the feasibility of this change, thanks.

@knottnt knottnt assigned sapphirew and michaelhtm and unassigned sapphirew Jul 21, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants