Skip to content

Release artifacts for release v1.5.1 - #64

Open
ack-bot wants to merge 1 commit into
mainfrom
ack-bot/release-v1.5.1
Open

Release artifacts for release v1.5.1#64
ack-bot wants to merge 1 commit into
mainfrom
ack-bot/release-v1.5.1

Conversation

@ack-bot

@ack-bot ack-bot commented Sep 8, 2026

Copy link
Copy Markdown
Collaborator

@ack-bot ack-bot added the prow/auto-gen PRs related to prow auto generation automation label Sep 8, 2026
@ack-prow

ack-prow Bot commented Sep 8, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: ack-bot
Once this PR has been reviewed and has the lgtm label, please assign knottnt for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@ack-prow
ack-prow Bot requested review from jlbutler and michaelhtm September 8, 2026 19:16
@ack-prow

ack-prow Bot commented Sep 8, 2026

Copy link
Copy Markdown

@ack-bot: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ram-verify-code-gen a4d1d12 link true /test ram-verify-code-gen

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

prow/auto-gen PRs related to prow auto generation automation

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant