Update dependency @testing-library/react to v16.3.3 - abandoned - #242
Open
renovate[bot] wants to merge 1 commit into
Open
Update dependency @testing-library/react to v16.3.3 - abandoned#242renovate[bot] wants to merge 1 commit into
renovate[bot] wants to merge 1 commit into
Conversation
🦋 Changeset detectedLatest commit: ff19124 The changes in this PR will be included in the next version bump. This PR includes changesets to release 2 packages
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
github-actions
Bot
force-pushed
the
renovate/patch-n-pin-updates
branch
from
August 25, 2026 15:38
296d1a0 to
b84a52b
Compare
renovate
Bot
force-pushed
the
renovate/patch-n-pin-updates
branch
from
August 28, 2026 13:38
b84a52b to
dd0f39b
Compare
renovate
Bot
force-pushed
the
renovate/patch-n-pin-updates
branch
from
August 28, 2026 19:41
dd0f39b to
3f83b53
Compare
github-actions
Bot
force-pushed
the
renovate/patch-n-pin-updates
branch
from
August 28, 2026 19:42
3f83b53 to
7c2210b
Compare
renovate
Bot
force-pushed
the
renovate/patch-n-pin-updates
branch
2 times, most recently
from
August 30, 2026 20:32
dd83927 to
1a09036
Compare
github-actions
Bot
force-pushed
the
renovate/patch-n-pin-updates
branch
from
August 30, 2026 20:33
1a09036 to
ff19124
Compare
Contributor
Author
Edited/Blocked NotificationRenovate will not automatically rebase this PR, because it does not recognize the last commit author and assumes somebody else may have edited the PR. You can manually request rebase by checking the rebase/retry box above. |
Contributor
Author
Autoclosing SkippedThis PR has been flagged for autoclosing. However, it is being skipped due to the branch being already modified. Please close/delete it manually or report a bug if you think this is in error. |
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
This PR contains the following updates:
16.3.2→16.3.3Release Notes
testing-library/react-testing-library (@testing-library/react)
v16.3.3Compare Source
Bug Fixes
Configuration
📅 Schedule: (in timezone Europe/Madrid)
🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.
🔕 Ignore: Close this PR and you won't be reminded about these updates again.
This PR was generated by Mend Renovate. View the repository job log.