add loadtest report sept 2026 - #4354
Conversation
|
/hold I think for this PR it makes sense to have the reviews both from @ntnn and @mjudeikis as you worked on a larger chunk of the fixes and I want to make sure I get it right |
|
/kind documentation |
On-behalf-of: SAP <simon.bein@sap.com> Signed-off-by: Simon Bein <simontheleg@gmail.com>
748ea59 to
c65daca
Compare
|
LGTM label has been added. DetailsGit tree hash: efdda0fcda777a5c3637fd463dd24bf942e1c0c2 |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: mjudeikis The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
|
/hold cancel I think we can merge and iterrate if needed |
|
@SimonTheLeg: The following test failed, say
DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here. |
Summary
Loadtest report of our efforts since April
What Type of PR Is This?
/kind documentation
Related Issue(s)
Fixes #platform-mesh/backlog#311
Release Notes