Skip to content

add glightbox to enlarge pics in documentation - #4356

Merged
kcp-ci-bot merged 1 commit into
kcp-dev:mainfrom
SimonTheLeg:add-glightbox
Sep 9, 2026
Merged

add glightbox to enlarge pics in documentation#4356
kcp-ci-bot merged 1 commit into
kcp-dev:mainfrom
SimonTheLeg:add-glightbox

Conversation

@SimonTheLeg

Copy link
Copy Markdown
Member

Summary

Allows one to click on images in our documentation to make them fullscreen

click on the image here
grafik

and then it becomes fullscreen in a lightbox
grafik

What Type of PR Is This?

/kind documentation

Related Issue(s)

Fixes #

Release Notes

NONE

On-behalf-of: SAP <simon.bein@sap.com>
Signed-off-by: Simon Bein <simontheleg@gmail.com>
@kcp-ci-bot kcp-ci-bot added release-note-none Denotes a PR that doesn't merit a release note. dco-signoff: yes Indicates the PR's author has signed the DCO. kind/documentation Categorizes issue or PR as related to documentation. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Sep 8, 2026
@mjudeikis

Copy link
Copy Markdown
Contributor

/lgtm
/approve

@kcp-ci-bot kcp-ci-bot added the lgtm Indicates that a PR is ready to be merged. label Sep 9, 2026
@kcp-ci-bot

Copy link
Copy Markdown
Contributor

LGTM label has been added.

DetailsGit tree hash: a96424f3778c9e0dea64559b33165ff406baa3f9

@kcp-ci-bot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: mjudeikis

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kcp-ci-bot kcp-ci-bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 9, 2026
@kcp-ci-bot
kcp-ci-bot merged commit 1765afd into kcp-dev:main Sep 9, 2026
6 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. dco-signoff: yes Indicates the PR's author has signed the DCO. kind/documentation Categorizes issue or PR as related to documentation. lgtm Indicates that a PR is ready to be merged. release-note-none Denotes a PR that doesn't merit a release note. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants