Skip to content

Revert "Fix regression in workspace deletion during migration" - #4357

Merged
mjudeikis merged 1 commit into
kcp-dev:mainfrom
mjudeikis:lc-migration-destination-local-client
Sep 10, 2026
Merged

Revert "Fix regression in workspace deletion during migration"#4357
mjudeikis merged 1 commit into
kcp-dev:mainfrom
mjudeikis:lc-migration-destination-local-client

Conversation

@mjudeikis

@mjudeikis mjudeikis commented Sep 10, 2026

Copy link
Copy Markdown
Contributor

Summary

This reverts commit 0524074.

I had my local setup configured to non existing push branch so this nevert happens again (git remote set-url --push upstream i-am-not-a-url) but dam :/
Thansk gh cli for fixing my local git config :/ sorry.

What Type of PR Is This?

Related Issue(s)

Fixes #

Release Notes

NONE

@kcp-ci-bot kcp-ci-bot added release-note-none Denotes a PR that doesn't merit a release note. do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. labels Sep 10, 2026
@kcp-ci-bot

Copy link
Copy Markdown
Contributor

Skipping CI for Draft Pull Request.
If you want CI signal for your change, please convert it to an actual PR.
You can still manually trigger a test run with /test all

@kcp-ci-bot kcp-ci-bot added dco-signoff: yes Indicates the PR's author has signed the DCO. do-not-merge/needs-kind Indicates a PR lacks a `kind/foo` label and requires one. size/M Denotes a PR that changes 30-99 lines, ignoring generated files. labels Sep 10, 2026
@mjudeikis
mjudeikis marked this pull request as ready for review September 10, 2026 08:49
@kcp-ci-bot kcp-ci-bot removed the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Sep 10, 2026
@mjudeikis mjudeikis added kind/bug Categorizes issue or PR as related to a bug. and removed do-not-merge/needs-kind Indicates a PR lacks a `kind/foo` label and requires one. labels Sep 10, 2026
@SimonTheLeg

Copy link
Copy Markdown
Member

/approve
/lgtm

@kcp-ci-bot kcp-ci-bot added the lgtm Indicates that a PR is ready to be merged. label Sep 10, 2026
@kcp-ci-bot

Copy link
Copy Markdown
Contributor

LGTM label has been added.

DetailsGit tree hash: dfc7efd6c7c85d484d6af2c68f070773d0324172

@kcp-ci-bot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: SimonTheLeg

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kcp-ci-bot kcp-ci-bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 10, 2026
@mjudeikis
mjudeikis merged commit f15baba into kcp-dev:main Sep 10, 2026
5 of 16 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. dco-signoff: yes Indicates the PR's author has signed the DCO. kind/bug Categorizes issue or PR as related to a bug. lgtm Indicates that a PR is ready to be merged. release-note-none Denotes a PR that doesn't merit a release note. size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants