Skip to content

Fix slate LC routing - #4359

Open
mjudeikis wants to merge 2 commits into
kcp-dev:mainfrom
mjudeikis:migration.fp.changes
Open

Fix slate LC routing#4359
mjudeikis wants to merge 2 commits into
kcp-dev:mainfrom
mjudeikis:migration.fp.changes

Conversation

@mjudeikis

Copy link
Copy Markdown
Contributor

Summary

During cluster migration we have bit of situation that front-proxy gets double events for same logical cluster on different shards. We need to explistlly handle copy object and to know where to route things :)

What Type of PR Is This?

/kind bug

Related Issue(s)

Fixes #

Release Notes

Better handle LC migration routing in front-proxy

@kcp-ci-bot kcp-ci-bot added release-note Denotes a PR that will be considered when it comes time to generate release notes. kind/bug Categorizes issue or PR as related to a bug. dco-signoff: yes Indicates the PR's author has signed the DCO. size/L Denotes a PR that changes 100-499 lines, ignoring generated files. labels Sep 10, 2026
@mjudeikis

Copy link
Copy Markdown
Contributor Author

/retest

@mjudeikis

Copy link
Copy Markdown
Contributor Author

/retest

@ntnn

ntnn commented Sep 11, 2026

Copy link
Copy Markdown
Member

MHmm... A cluster that is being migrated shouldn't be handled by front-proxy at all until the migration is done.

/hold

@kcp-ci-bot kcp-ci-bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Sep 11, 2026
@mjudeikis

Copy link
Copy Markdown
Contributor Author

Im on the line... There is still routing happening as shard responds with "its not available"... And the problem is that we now get 2 events from 2 different shards for same LC.

@mjudeikis

Copy link
Copy Markdown
Contributor Author

I think informers will still emit events? Movign this conv to slack.

@mjudeikis

Copy link
Copy Markdown
Contributor Author

/retest

@kcp-ci-bot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign mjudeikis for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@mjudeikis

Copy link
Copy Markdown
Contributor Author

/retest
flake #4362

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

dco-signoff: yes Indicates the PR's author has signed the DCO. do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. kind/bug Categorizes issue or PR as related to a bug. release-note Denotes a PR that will be considered when it comes time to generate release notes. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants