Skip to content

Bump stale schema - #4360

Merged
kcp-ci-bot merged 1 commit into
kcp-dev:mainfrom
mjudeikis:fix.schema
Sep 10, 2026
Merged

kcp-ci-bot merged 1 commit into
kcp-dev:mainfrom
mjudeikis:fix.schema

Conversation

@mjudeikis

Copy link
Copy Markdown
Contributor

Summary

Somehow we got stale scehma in the main. Meaning schema was updated, but not the name. Upgrade test now fails

What Type of PR Is This?

/kind bug

Related Issue(s)

Fixes #

Release Notes

NONE

@kcp-ci-bot kcp-ci-bot added release-note-none Denotes a PR that doesn't merit a release note. kind/bug Categorizes issue or PR as related to a bug. dco-signoff: yes Indicates the PR's author has signed the DCO. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Sep 10, 2026

@ntnn ntnn left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm
/approve

I feel a dejavu. Didn't we have this recently?

@kcp-ci-bot kcp-ci-bot added the lgtm Indicates that a PR is ready to be merged. label Sep 10, 2026
@kcp-ci-bot

Copy link
Copy Markdown
Contributor

LGTM label has been added.

DetailsGit tree hash: 6cd5ad891fd75f086875b7a94a8f245d3f3081c7

@kcp-ci-bot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: ntnn

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kcp-ci-bot kcp-ci-bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 10, 2026
@kcp-ci-bot
kcp-ci-bot merged commit 01dd585 into kcp-dev:main Sep 10, 2026
17 checks passed
@mjudeikis

Copy link
Copy Markdown
Contributor Author

We did. Upgrade test should catch these now

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. dco-signoff: yes Indicates the PR's author has signed the DCO. kind/bug Categorizes issue or PR as related to a bug. lgtm Indicates that a PR is ready to be merged. release-note-none Denotes a PR that doesn't merit a release note. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants