ci(packit): run Koji/Bodhi on production - #1016
Merged
Merged
Conversation
Contributor
|
Build succeeded. ✔️ pre-commit SUCCESS in 3m 19s |
betulependule
approved these changes
Sep 14, 2026
betulependule
left a comment
Contributor
There was a problem hiding this comment.
LGTM. Bodhi update isn't part of ogr's .packit.yaml so maybe I would consider removing the mention of it from the title of the PR and the commit?
Move Koji automation to the production, so that we are able to test the Forgejo dist-git deployment before it gets officially moved. Signed-off-by: Matej Focko <mfocko@packit.dev>
mfocko
force-pushed
the
ci/adjust-downstream-ci
branch
from
September 14, 2026 07:56
cd51d6b to
8e30fd2
Compare
Contributor
|
Build succeeded. ✔️ pre-commit SUCCESS in 3m 16s |
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Move Koji/Bodhi automation to the production, so that we are able to test the Forgejo dist-git deployment before it gets officially moved.
TODO: