Add Pipelines v1.15.x to website sync configuration - #734
Merged
Conversation
vdemeester
force-pushed
the
add-v1.15-sync
branch
from
July 31, 2026 13:24
0b4415a to
3431704
Compare
Member
|
linter issues will be fixed by #736 |
Member
Author
|
ah I need to rebase, thanks @AlanGreene |
vdemeester
force-pushed
the
add-v1.15-sync
branch
from
August 31, 2026 13:18
3431704 to
4e25a49
Compare
|
[APPROVALNOTIFIER] This PR is APPROVED Approval requirements bypassed by manually added approval. This pull-request has been approved by: The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Changes
Add
release-v1.15.xbranch to the Pipelines sync configuration so that v1.15.x docs are published to tekton.dev.Submitter Checklist