Skip to content

update rocksdb (rocksdb#437) in tikv-7.5 - #859

Merged
ti-chi-bot[bot] merged 1 commit into
tikv:tikv-7.5from
gengliqi:fix-ingest-allow-write-7.5
Aug 12, 2026
Merged

ti-chi-bot[bot] merged 1 commit into
tikv:tikv-7.5from
gengliqi:fix-ingest-allow-write-7.5

Conversation

@gengliqi

Copy link
Copy Markdown
Member

Update rocksdb to include the bugfix tikv/rocksdb#437

Signed-off-by: gengliqi <gengliqiii@gmail.com>
@ti-chi-bot ti-chi-bot Bot added the dco-signoff: yes Indicates the PR's author has signed the dco. label Aug 12, 2026
@coderabbitai

coderabbitai Bot commented Aug 12, 2026

Copy link
Copy Markdown

Important

Review skipped

Auto reviews are disabled on base/target branches other than the default branch.

Please check the settings in the CodeRabbit UI or the .coderabbit.yaml file in this repository. To trigger a single review, invoke the @coderabbitai review command.

⚙️ Run configuration

Configuration used: defaults

Review profile: CHILL

Plan: Pro Plus

Run ID: 65f02ce3-2cc3-4bee-b372-90f6c010d7f5

You can disable this status message by setting the reviews.review_status to false in the CodeRabbit configuration file.

Use the checkbox below for a quick retry:

  • 🔍 Trigger review

Thanks for using CodeRabbit! It's free for OSS, and your support helps us grow. If you like it, consider giving us a shout-out.

❤️ Share

Comment @coderabbitai help to get the list of available commands.

@gengliqi gengliqi changed the title update rocksdb (rocksdb#437) update rocksdb (rocksdb#437) in tikv-7.5 Aug 12, 2026
@ti-chi-bot ti-chi-bot Bot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Aug 12, 2026
@ti-chi-bot

ti-chi-bot Bot commented Aug 12, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: overvenus

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@ti-chi-bot ti-chi-bot Bot added the lgtm label Aug 12, 2026
@ti-chi-bot

ti-chi-bot Bot commented Aug 12, 2026

Copy link
Copy Markdown

[LGTM Timeline notifier]

Timeline:

  • 2026-08-12 06:29:07.171299789 +0000 UTC m=+3200733.207394845: ☑️ agreed by overvenus.

@ti-chi-bot ti-chi-bot Bot added the approved label Aug 12, 2026
@ti-chi-bot
ti-chi-bot Bot merged commit b3f1ecb into tikv:tikv-7.5 Aug 12, 2026
4 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved dco-signoff: yes Indicates the PR's author has signed the dco. lgtm size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants