Skip to content

Added Iam binding list support - #18925

Open
malhotrasagar2212 wants to merge 5 commits into
GoogleCloudPlatform:mainfrom
malhotrasagar2212:iam_binding_list
Open

Added Iam binding list support#18925
malhotrasagar2212 wants to merge 5 commits into
GoogleCloudPlatform:mainfrom
malhotrasagar2212:iam_binding_list

Conversation

@malhotrasagar2212

Copy link
Copy Markdown
Collaborator

Release Note Template for Downstream PRs (will be copied)
Added IAM binf list support and list resource for google_project_iam_binding
See Write release notes for guidance.

`google_project_iam_binding`

@modular-magician

modular-magician commented Sep 8, 2026

Copy link
Copy Markdown
Collaborator

Hi there, I'm the Modular magician. I've detected the following information about your changes for commit e82231e:

Diff report

Your PR generated the following diffs in downstream repositories:

Repository Diff Link Changes
google provider View Diff 5 files changed, 653 insertions(+), 206 deletions(-)
google-beta provider View Diff 5 files changed, 653 insertions(+), 206 deletions(-)
terraform-google-conversion View Diff 1 file changed, 24 insertions(+)

Test report

Analytics

Total Tests Passed Skipped Affected
6984 6268 698 18
Affected Service Packages
  • All service packages are affected

Learn how VCR tests work


Step 1: Replaying Mode

Action taken

Found 18 affected test(s) by replaying old test recordings. Starting RECORDING based on the most recent commit.

Click here to see the affected tests
  • TestAccComputeBackendService_backendServiceInFlightExample
  • TestAccComputeInstanceFromMachineImage_diffProject
  • TestAccComputeInstanceTemplate_gracefulShutdown
  • TestAccComputeInstance_GracefulShutdownWithResetUpdate
  • TestAccComputeInstance_GracefulShutdownWithoutResetUpdate
  • TestAccComputeRegionInstanceTemplate_gracefulShutdown
  • TestAccComputeSecurityPolicy_withPreconfiguredWafConfig
  • TestAccComputeTargetInstance_withSecurityPolicy
  • TestAccContainerCluster_autopilot_privileged_admission
  • TestAccContainerCluster_dataplaneOptimizationMode
  • TestAccContainerNodePool_withSoleTenantConfig
  • TestAccDataSourceGoogleComputeInterconnectLocation_basic
  • TestAccDataSourceGoogleStorageControlOrganizationIntelligenceFindingsSummary_empty
  • TestAccDataprocMetastoreService_dataprocMetastoreServicePrivateServiceConnectExample
  • TestAccDiscoveryEngineDataStore_discoveryengineDatastoreKmsKeyNameExample
  • TestAccPubsubSubscription_pubsubSubscriptionTagsExample
  • TestAccTpuV2Vm_update
  • TestAccVertexAIFeatureOnlineStoreFeatureview_vertexAiFeatureonlinestoreFeatureview_featureRegistry_updated

View the replaying VCR build log


Step 2: Recording Mode

Recording Mode Replaying Rerun Test Name
❌ Error · Log - TestAccComputeBackendService_backendServiceInFlightExample
❌ Error · Log - TestAccComputeInstanceFromMachineImage_diffProject
❌ Error · Log - TestAccComputeInstanceTemplate_gracefulShutdown
❌ Error · Log - TestAccComputeInstance_GracefulShutdownWithResetUpdate
❌ Error · Log - TestAccComputeInstance_GracefulShutdownWithoutResetUpdate
❌ Error · Log - TestAccComputeRegionInstanceTemplate_gracefulShutdown
❌ Error · Log - TestAccComputeSecurityPolicy_withPreconfiguredWafConfig
❌ Error · Log - TestAccComputeTargetInstance_withSecurityPolicy
❌ Error · Log - TestAccContainerCluster_autopilot_privileged_admission
❌ Error · Log - TestAccContainerCluster_dataplaneOptimizationMode
❌ Error · Log - TestAccContainerNodePool_withSoleTenantConfig
❌ Error · Log - TestAccDataSourceGoogleComputeInterconnectLocation_basic
❌ Error · Log - TestAccDataSourceGoogleStorageControlOrganizationIntelligenceFindingsSummary_empty
❌ Error · Log - TestAccDataprocMetastoreService_dataprocMetastoreServicePrivateServiceConnectExample
❌ Error · Log - TestAccDiscoveryEngineDataStore_discoveryengineDatastoreKmsKeyNameExample
❌ Error · Log - TestAccPubsubSubscription_pubsubSubscriptionTagsExample
❌ Error · Log - TestAccTpuV2Vm_update
❌ Error · Log - TestAccVertexAIFeatureOnlineStoreFeatureview_vertexAiFeatureonlinestoreFeatureview_featureRegistry_updated

Caution

Issues requiring attention before PR completion

🔴 Initial Recording Failed: Some tests failed during the recording step. See the table above for details.

Please address these issues to complete your PR. If you believe these detections are incorrect or unrelated to your change, please raise the concern with your reviewer.

View the recording VCR build log or the debug logs folder for detailed results.

@malhotrasagar2212 VCR tests complete for e82231e!

@modular-magician

modular-magician commented Sep 8, 2026

Copy link
Copy Markdown
Collaborator

Hi there, I'm the Modular magician. I've detected the following information about your changes for commit 045bc79:

Diff report

Your PR generated the following diffs in downstream repositories:

Repository Diff Link Changes
google provider View Diff 5 files changed, 445 insertions(+), 11 deletions(-)
google-beta provider View Diff 5 files changed, 445 insertions(+), 11 deletions(-)
terraform-google-conversion View Diff 1 file changed, 24 insertions(+)

Test report

Analytics

Total Tests Passed Skipped Affected
6988 6273 698 17
Affected Service Packages
  • All service packages are affected

Learn how VCR tests work


Step 1: Replaying Mode

Action taken

Found 17 affected test(s) by replaying old test recordings. Starting RECORDING based on the most recent commit.

Click here to see the affected tests
  • TestAccComputeBackendService_backendServiceInFlightExample
  • TestAccComputeInstanceFromMachineImage_diffProject
  • TestAccComputeInstanceTemplate_gracefulShutdown
  • TestAccComputeInstance_GracefulShutdownWithResetUpdate
  • TestAccComputeInstance_GracefulShutdownWithoutResetUpdate
  • TestAccComputeRegionInstanceTemplate_gracefulShutdown
  • TestAccComputeSecurityPolicy_withPreconfiguredWafConfig
  • TestAccComputeTargetInstance_withSecurityPolicy
  • TestAccContainerCluster_autopilot_privileged_admission
  • TestAccContainerCluster_dataplaneOptimizationMode
  • TestAccContainerNodePool_withSoleTenantConfig
  • TestAccDataSourceGoogleStorageControlOrganizationIntelligenceFindingsSummary_empty
  • TestAccDataprocMetastoreService_dataprocMetastoreServicePrivateServiceConnectExample
  • TestAccDiscoveryEngineDataStore_discoveryengineDatastoreKmsKeyNameExample
  • TestAccPubsubSubscription_pubsubSubscriptionTagsExample
  • TestAccTpuV2Vm_update
  • TestAccVertexAIFeatureOnlineStoreFeatureview_vertexAiFeatureonlinestoreFeatureview_featureRegistry_updated

View the replaying VCR build log


Step 2: Recording Mode

Recording Mode Replaying Rerun Test Name
❌ Error · Log - TestAccComputeBackendService_backendServiceInFlightExample
❌ Error · Log - TestAccComputeInstanceFromMachineImage_diffProject
❌ Error · Log - TestAccComputeInstanceTemplate_gracefulShutdown
❌ Error · Log - TestAccComputeInstance_GracefulShutdownWithResetUpdate
❌ Error · Log - TestAccComputeInstance_GracefulShutdownWithoutResetUpdate
❌ Error · Log - TestAccComputeRegionInstanceTemplate_gracefulShutdown
❌ Error · Log - TestAccComputeSecurityPolicy_withPreconfiguredWafConfig
❌ Error · Log - TestAccComputeTargetInstance_withSecurityPolicy
❌ Error · Log - TestAccContainerCluster_autopilot_privileged_admission
❌ Error · Log - TestAccContainerCluster_dataplaneOptimizationMode
❌ Error · Log - TestAccContainerNodePool_withSoleTenantConfig
❌ Error · Log - TestAccDataSourceGoogleStorageControlOrganizationIntelligenceFindingsSummary_empty
❌ Error · Log - TestAccDataprocMetastoreService_dataprocMetastoreServicePrivateServiceConnectExample
❌ Error · Log - TestAccDiscoveryEngineDataStore_discoveryengineDatastoreKmsKeyNameExample
❌ Error · Log - TestAccPubsubSubscription_pubsubSubscriptionTagsExample
❌ Error · Log - TestAccTpuV2Vm_update
❌ Error · Log - TestAccVertexAIFeatureOnlineStoreFeatureview_vertexAiFeatureonlinestoreFeatureview_featureRegistry_updated

Caution

Issues requiring attention before PR completion

🔴 Initial Recording Failed: Some tests failed during the recording step. See the table above for details.

Please address these issues to complete your PR. If you believe these detections are incorrect or unrelated to your change, please raise the concern with your reviewer.

View the recording VCR build log or the debug logs folder for detailed results.

@malhotrasagar2212 VCR tests complete for 045bc79!

@modular-magician

modular-magician commented Sep 8, 2026

Copy link
Copy Markdown
Collaborator

Hi there, I'm the Modular magician. I've detected the following information about your changes for commit 32c594b:

Diff report

Your PR generated the following diffs in downstream repositories:

Repository Diff Link Changes
google provider View Diff 5 files changed, 446 insertions(+), 12 deletions(-)
google-beta provider View Diff 5 files changed, 446 insertions(+), 12 deletions(-)
terraform-google-conversion View Diff 1 file changed, 24 insertions(+)

Test report

Analytics

Total Tests Passed Skipped Affected
6988 6271 698 19
Affected Service Packages
  • All service packages are affected

Learn how VCR tests work


Step 1: Replaying Mode

Action taken

Found 19 affected test(s) by replaying old test recordings. Starting RECORDING based on the most recent commit.

Click here to see the affected tests
  • TestAccComputeBackendService_backendServiceInFlightExample
  • TestAccComputeInstanceFromMachineImage_diffProject
  • TestAccComputeInstanceTemplate_gracefulShutdown
  • TestAccComputeInstance_GracefulShutdownWithResetUpdate
  • TestAccComputeInstance_GracefulShutdownWithoutResetUpdate
  • TestAccComputeRegionInstanceTemplate_gracefulShutdown
  • TestAccComputeSecurityPolicy_withPreconfiguredWafConfig
  • TestAccComputeTargetInstance_withSecurityPolicy
  • TestAccContainerCluster_autopilot_privileged_admission
  • TestAccContainerCluster_dataplaneOptimizationMode
  • TestAccContainerNodePool_withSoleTenantConfig
  • TestAccDataSourceGoogleStorageControlOrganizationIntelligenceFindingsSummary_empty
  • TestAccDataformConfig_update
  • TestAccDataprocMetastoreService_dataprocMetastoreServicePrivateServiceConnectExample
  • TestAccDiscoveryEngineDataStore_discoveryengineDatastoreKmsKeyNameExample
  • TestAccFolderIamPolicy_basic
  • TestAccPubsubSubscription_pubsubSubscriptionTagsExample
  • TestAccTpuV2Vm_update
  • TestAccVertexAIFeatureOnlineStoreFeatureview_vertexAiFeatureonlinestoreFeatureview_featureRegistry_updated

View the replaying VCR build log


Step 2: Recording Mode

Recording Mode Replaying Rerun Test Name
✅ Log TestAccDataformConfig_update
✅ Log TestAccFolderIamPolicy_basic
❌ Error · Log - TestAccComputeBackendService_backendServiceInFlightExample
❌ Error · Log - TestAccComputeInstanceFromMachineImage_diffProject
❌ Error · Log - TestAccComputeInstanceTemplate_gracefulShutdown
❌ Error · Log - TestAccComputeInstance_GracefulShutdownWithResetUpdate
❌ Error · Log - TestAccComputeInstance_GracefulShutdownWithoutResetUpdate
❌ Error · Log - TestAccComputeRegionInstanceTemplate_gracefulShutdown
❌ Error · Log - TestAccComputeSecurityPolicy_withPreconfiguredWafConfig
❌ Error · Log - TestAccComputeTargetInstance_withSecurityPolicy
❌ Error · Log - TestAccContainerCluster_autopilot_privileged_admission
❌ Error · Log - TestAccContainerCluster_dataplaneOptimizationMode
❌ Error · Log - TestAccContainerNodePool_withSoleTenantConfig
❌ Error · Log - TestAccDataSourceGoogleStorageControlOrganizationIntelligenceFindingsSummary_empty
❌ Error · Log - TestAccDataprocMetastoreService_dataprocMetastoreServicePrivateServiceConnectExample
❌ Error · Log - TestAccDiscoveryEngineDataStore_discoveryengineDatastoreKmsKeyNameExample
❌ Error · Log - TestAccPubsubSubscription_pubsubSubscriptionTagsExample
❌ Error · Log - TestAccTpuV2Vm_update
❌ Error · Log - TestAccVertexAIFeatureOnlineStoreFeatureview_vertexAiFeatureonlinestoreFeatureview_featureRegistry_updated

Caution

Issues requiring attention before PR completion

🔴 Initial Recording Failed: Some tests failed during the recording step. See the table above for details.

Please address these issues to complete your PR. If you believe these detections are incorrect or unrelated to your change, please raise the concern with your reviewer.

View the recording VCR build log or the debug logs folder for detailed results.

@malhotrasagar2212 VCR tests complete for 32c594b!

@malhotrasagar2212
malhotrasagar2212 marked this pull request as ready for review September 9, 2026 15:00
@BBBmau

BBBmau commented Sep 9, 2026

Copy link
Copy Markdown
Collaborator

for some reason TestAccProjectIamBindingListResource_queryIdentity isn't being ran in VCR?

@BBBmau

BBBmau commented Sep 9, 2026

Copy link
Copy Markdown
Collaborator

/gcbrun

@modular-magician

modular-magician commented Sep 9, 2026

Copy link
Copy Markdown
Collaborator

Hi there, I'm the Modular magician. I've detected the following information about your changes for commit 32c594b:

Diff report

Your PR generated the following diffs in downstream repositories:

Repository Diff Link Changes
google provider View Diff 5 files changed, 446 insertions(+), 12 deletions(-)
google-beta provider View Diff 5 files changed, 446 insertions(+), 12 deletions(-)
terraform-google-conversion View Diff 1 file changed, 24 insertions(+)

Test report

Analytics

Total Tests Passed Skipped Affected
6989 6271 698 20
Affected Service Packages
  • All service packages are affected

Learn how VCR tests work


Step 1: Replaying Mode

Action taken

Found 20 affected test(s) by replaying old test recordings. Starting RECORDING based on the most recent commit.

Click here to see the affected tests
  • TestAccComputeBackendService_backendServiceInFlightExample
  • TestAccComputeInstanceFromMachineImage_diffProject
  • TestAccComputeInstanceTemplate_gracefulShutdown
  • TestAccComputeInstance_GracefulShutdownWithResetUpdate
  • TestAccComputeInstance_GracefulShutdownWithoutResetUpdate
  • TestAccComputeRegionInstanceTemplate_gracefulShutdown
  • TestAccComputeSecurityPolicy_withPreconfiguredWafConfig
  • TestAccComputeTargetInstance_withSecurityPolicy
  • TestAccContainerCluster_autopilot_privileged_admission
  • TestAccContainerCluster_dataplaneOptimizationMode
  • TestAccContainerNodePool_withSoleTenantConfig
  • TestAccDataSourceGoogleStorageControlOrganizationIntelligenceFindingsSummary_empty
  • TestAccDataformConfig_update
  • TestAccDataprocMetastoreService_dataprocMetastoreServicePrivateServiceConnectExample
  • TestAccDiscoveryEngineDataStore_discoveryengineDatastoreKmsKeyNameExample
  • TestAccFirebaseAppCheckAppAttestConfig_firebaseAppCheckAppAttestConfigUpdate
  • TestAccObservabilityLink_observabilityLinkBasicExample
  • TestAccPubsubSubscription_pubsubSubscriptionTagsExample
  • TestAccTpuV2Vm_update
  • TestAccVertexAIFeatureOnlineStoreFeatureview_vertexAiFeatureonlinestoreFeatureview_featureRegistry_updated

View the replaying VCR build log


Step 2: Recording Mode

Recording Mode Replaying Rerun Test Name
✅ Log TestAccDataformConfig_update
✅ Log TestAccFirebaseAppCheckAppAttestConfig_firebaseAppCheckAppAttestConfigUpdate
✅ Log ❌ Error · Log TestAccObservabilityLink_observabilityLinkBasicExample
❌ Error · Log - TestAccComputeBackendService_backendServiceInFlightExample
❌ Error · Log - TestAccComputeInstanceFromMachineImage_diffProject
❌ Error · Log - TestAccComputeInstanceTemplate_gracefulShutdown
❌ Error · Log - TestAccComputeInstance_GracefulShutdownWithResetUpdate
❌ Error · Log - TestAccComputeInstance_GracefulShutdownWithoutResetUpdate
❌ Error · Log - TestAccComputeRegionInstanceTemplate_gracefulShutdown
❌ Error · Log - TestAccComputeSecurityPolicy_withPreconfiguredWafConfig
❌ Error · Log - TestAccComputeTargetInstance_withSecurityPolicy
❌ Error · Log - TestAccContainerCluster_autopilot_privileged_admission
❌ Error · Log - TestAccContainerCluster_dataplaneOptimizationMode
❌ Error · Log - TestAccContainerNodePool_withSoleTenantConfig
❌ Error · Log - TestAccDataSourceGoogleStorageControlOrganizationIntelligenceFindingsSummary_empty
❌ Error · Log - TestAccDataprocMetastoreService_dataprocMetastoreServicePrivateServiceConnectExample
❌ Error · Log - TestAccDiscoveryEngineDataStore_discoveryengineDatastoreKmsKeyNameExample
❌ Error · Log - TestAccPubsubSubscription_pubsubSubscriptionTagsExample
❌ Error · Log - TestAccTpuV2Vm_update
❌ Error · Log - TestAccVertexAIFeatureOnlineStoreFeatureview_vertexAiFeatureonlinestoreFeatureview_featureRegistry_updated

Caution

Issues requiring attention before PR completion

🔴 Initial Recording Failed: Some tests failed during the recording step. See the table above for details.

🔴 Replaying Rerun Failed: Some tests failed due to non-determinism when VCR replayed the response. See the table above for details.

Please address these issues to complete your PR. If you believe these detections are incorrect or unrelated to your change, please raise the concern with your reviewer.

View the recording VCR build log or the debug logs folder for detailed results.

@malhotrasagar2212, @BBBmau VCR tests complete for 32c594b!

@malhotrasagar2212

malhotrasagar2212 commented Sep 9, 2026

Copy link
Copy Markdown
Collaborator Author

@BBBmau here is the log for TestAccProjectIamBindingListResource_queryIdentity VCR run.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants