VAP/MAP improvements - #4217
VAP/MAP improvements#4217
Conversation
|
They all fail on TestMutatingAdmissionPolicyCrossWorkspaceAPIBinding - I'm thinking the sync wait is the problem. |
|
LGTM label has been added. DetailsGit tree hash: 08b25f161b104c0f67522b6ae5f34487a9447137 |
|
/retest |
f249b7a to
b0f6029
Compare
Signed-off-by: Nelo-T. Wallus <n.wallus@sap.com>
b0f6029 to
1416471
Compare
1416471 to
ac3bd6e
Compare
|
LGTM label has been added. DetailsGit tree hash: 5244d3843c194f9f8a053ba0c1ecc6c0211a1561 |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: mjudeikis The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
|
/retest Flake adressed by #4219 |
|
/retest metrics endpoint flake |
|
/cherrypick release-0.32 |
|
@SimonTheLeg: new pull request created: #4276 DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
Summary
Thank you for noticing the branch name. Yes, I am a comedic genius.
source policytarget cluster bounded by its per-cluster contextDoesn't quite fix #4216 but kinda does.
What Type of PR Is This?
/kind bug
/kind feature
Related Issue(s)
Fixes #
Release Notes