Skip to content

[release-0.32] VAP/MAP improvements - #4276

Merged
kcp-ci-bot merged 9 commits into
kcp-dev:release-0.32from
kcp-ci-bot:cherry-pick-4217-to-release-0.32
Jul 16, 2026
Merged

kcp-ci-bot merged 9 commits into
kcp-dev:release-0.32from
kcp-ci-bot:cherry-pick-4217-to-release-0.32

Conversation

@kcp-ci-bot

Copy link
Copy Markdown
Contributor

This is an automated cherry-pick of #4217

/assign SimonTheLeg

NONE

@kcp-ci-bot kcp-ci-bot added the release-note-none Denotes a PR that doesn't merit a release note. label Jul 16, 2026
@kcp-ci-bot kcp-ci-bot mentioned this pull request Jul 16, 2026
@kcp-ci-bot kcp-ci-bot added dco-signoff: yes Indicates the PR's author has signed the DCO. kind/api-change Categorizes issue or PR as related to adding, removing, or otherwise changing an API size/L Denotes a PR that changes 100-499 lines, ignoring generated files. labels Jul 16, 2026
@SimonTheLeg
SimonTheLeg requested a review from ntnn July 16, 2026 08:48

@ntnn ntnn left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm
/approve

@kcp-ci-bot kcp-ci-bot added the lgtm Indicates that a PR is ready to be merged. label Jul 16, 2026
@kcp-ci-bot

Copy link
Copy Markdown
Contributor Author

LGTM label has been added.

DetailsGit tree hash: e131e5c49e75860d2855119c3cd75889f5055cb2

@xrstf

xrstf commented Jul 16, 2026

Copy link
Copy Markdown
Contributor

/approve

@kcp-ci-bot

Copy link
Copy Markdown
Contributor Author

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: ntnn, xrstf

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kcp-ci-bot kcp-ci-bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jul 16, 2026
@kcp-ci-bot
kcp-ci-bot merged commit 4a47e3b into kcp-dev:release-0.32 Jul 16, 2026
14 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. dco-signoff: yes Indicates the PR's author has signed the DCO. kind/api-change Categorizes issue or PR as related to adding, removing, or otherwise changing an API lgtm Indicates that a PR is ready to be merged. release-note-none Denotes a PR that doesn't merit a release note. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants